Skip to content

[release-4.20] SPLAT-2867: Add vSphere multi vCenter Day 2 support - #10880

Open
vr4manta wants to merge 5 commits into
openshift:release-4.20from
vr4manta:SPLAT-2867
Open

vr4manta wants to merge 5 commits into
openshift:release-4.20from
vr4manta:SPLAT-2867

Conversation

@vr4manta

@vr4manta vr4manta commented Sep 16, 2026 •

Copy link
Copy Markdown
Contributor

SPLAT-2867

Changes

  • Changed package used for generating vSphere cloud config to match 3CMO so there are no differences (prevents MCO from thinking bootstrap config doesn' t match initial MC generated at operator's first start)
  • Enhanced cloud config generation logic to include node network cidr information
  • Removed FeatureGateVSphereMultiNetworks check since its been GA for a while now
  • Modified vSphere cloud config to not set port for vCenter if not configured in install-config
  • Ported library-go logic into the installer due to go.mod dependency issue with pinned dependencies
  • Added feature gate check of FeatureGateVSphereMultiNetworks in vsphere/cloudproviderconfig.go when generating node network configs

Notes

This PR is backporting the following PRs in relation to the vSphere multi vCenter day 2 support feature:

In addition to the two PRs above, we needed to wrap the multi network logic in the cloudproviderconfig module to not generate the internal/external machine cidr info into the cloud config. This would cause an issue if running CNU w/ day 2 enabled but not the FeatureGateVSphereMultiNetworks. Full TPNU was fine.

@openshift-merge-bot

Copy link
Copy Markdown
Contributor

Pipeline controller notification
This repo is configured to use the pipeline controller. Second-stage tests will be triggered either automatically or after lgtm label is added, depending on the repository configuration. The pipeline controller will automatically detect which contexts are required and will utilize /test Prow commands to trigger the second stage.

For optional jobs, comment /test ? to see a list of all defined jobs. To trigger manually all jobs from second stage use /pipeline required command.

This repository is configured in: LGTM mode

@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Sep 16, 2026
@openshift-ci openshift-ci Bot added the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Sep 16, 2026
@openshift-ci

openshift-ci Bot commented Sep 16, 2026

Copy link
Copy Markdown
Contributor

Skipping CI for Draft Pull Request.
If you want CI signal for your change, please convert it to an actual PR.
You can still manually trigger a test run with /test all

@openshift-ci-robot

openshift-ci-robot commented Sep 16, 2026 •

Copy link
Copy Markdown
Contributor

@vr4manta: This pull request references SPLAT-2867 which is a valid jira issue.

Details

In response to this:

SPLAT-2867

Changes

  • Changed package used for generating vSphere cloud config to match 3CMO so there are no differences (prevents MCO from thinking bootstrap config doesn' t match initial MC generated at operator's first start)
  • Enhanced cloud config generation logic to include node network cidr information
  • Removed FeatureGateVSphereMultiNetworks check since its been GA for a while now
  • Modified vSphere cloud config to not set port for vCenter if not configured in install-config
  • Updated go.mod to use fork of openshift/api and openshift/library-go due to issues with changes in those repos causing incompatibility with current installer configs

Notes

This PR is backporting the following PRs in relation to the vSphere multi vCenter day 2 support feature:

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@coderabbitai

coderabbitai Bot commented Sep 16, 2026 •

Copy link
Copy Markdown

Important

Review skipped

Auto reviews are disabled on base/target branches other than the default branch.

Please check the settings in the CodeRabbit UI or the .coderabbit.yaml file in this repository. To trigger a single review, invoke the @coderabbitai review command.

⚙️ Run configuration

Configuration used: Repository: openshift/coderabbit/.coderabbit.yaml

Review profile: CHILL

Plan: Enterprise

Run ID: 83b1632f-62fa-4002-aa94-cf094c52cb46

You can disable this status message by setting the reviews.review_status to false in the CodeRabbit configuration file.

Use the checkbox below for a quick retry:

  • 🔍 Trigger review

Comment @coderabbitai help to get the list of available commands.

@vr4manta vr4manta changed the title WIP: [release-4.20] SPLAT-2867: Add vSphere multi vCenter Day 2 support [release-4.20] SPLAT-2867: Add vSphere multi vCenter Day 2 support Sep 30, 2026
@vr4manta
vr4manta marked this pull request as ready for review September 30, 2026 17:41
@openshift-ci openshift-ci Bot removed the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Sep 30, 2026
@vr4manta
vr4manta force-pushed the SPLAT-2867 branch 3 times, most recently from edb4db2 to 4107a4f Compare October 1, 2026 15:55
@vr4manta

vr4manta commented Oct 5, 2026

Copy link
Copy Markdown
Contributor Author

/payload-job-with-prs periodic-ci-openshift-release-main-nightly-4.21-e2e-vsphere-ovn-zones-techpreview-serial openshift/cluster-config-operator#506 openshift/cluster-cloud-controller-manager-operator#521 openshift/machine-api-operator#1555

@openshift-ci

openshift-ci Bot commented Oct 5, 2026

Copy link
Copy Markdown
Contributor

@vr4manta: trigger 2 job(s) for the /payload-(with-prs|job|aggregate|job-with-prs|aggregate-with-prs) command

  • periodic-ci-openshift-release-main-nightly-4.21-e2e-vsphere-ovn-zones-techpreview-serial-1of2
  • periodic-ci-openshift-release-main-nightly-4.21-e2e-vsphere-ovn-zones-techpreview-serial-2of2

See details on https://pr-payload-tests.ci.openshift.org/runs/ci/e3cc9fe0-c0e8-11f1-94df-e062ca21626b-0

@rvanderp3

Copy link
Copy Markdown
Contributor

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Oct 5, 2026
@openshift-merge-bot

Copy link
Copy Markdown
Contributor

Scheduling tests matching the pipeline_run_if_changed or not excluded by pipeline_skip_if_only_changed parameters:
/test e2e-vsphere-multi-vcenter-ovn
/test e2e-vsphere-ovn
/test e2e-vsphere-ovn-disk-setup-techpreview
/test e2e-vsphere-ovn-hybrid-env
/test e2e-vsphere-ovn-techpreview
/test e2e-vsphere-ovn-zones
/test e2e-aws-ovn

@jcpowermac

Copy link
Copy Markdown
Contributor

/approve

@openshift-ci

openshift-ci Bot commented Oct 5, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: jcpowermac

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Oct 5, 2026
@vr4manta

vr4manta commented Oct 6, 2026

Copy link
Copy Markdown
Contributor Author

/retest

@openshift-ci

openshift-ci Bot commented Oct 6, 2026

Copy link
Copy Markdown
Contributor

@vr4manta: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants