Skip to content

1.5.0 - #190

Merged
pablopunk merged 1 commit into
mainfrom
release/1.5.0
Sep 18, 2026
Merged

1.5.0#190
pablopunk merged 1 commit into
mainfrom
release/1.5.0

Conversation

@pablopunk

Copy link
Copy Markdown
Owner

Release 1.5.0

@vercel

vercel Bot commented Sep 18, 2026 •

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

1 Skipped Deployment
Project Deployment Actions Updated
swiftshift-app Ignored Ignored Preview Sep 18, 2026 8:43am UTC

@pablopunk
pablopunk merged commit 0b96725 into main Sep 18, 2026
1 check was pending
@codesandbox

codesandbox Bot commented Sep 18, 2026

Copy link
Copy Markdown

Review or Edit in CodeSandbox

Open the branch in Web Editor • VS Code • Insiders

Open Preview

@pablopunk
pablopunk deleted the release/1.5.0 branch September 18, 2026 08:43
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant