Skip to content

[main] Update actions/deploy-pages digest to 368f825 (main) - #504

Merged
openshift-merge-bot[bot] merged 1 commit into
mainfrom
renovate/main-github-actions-dependencies
Sep 7, 2026
Merged

[main] Update actions/deploy-pages digest to 368f825 (main)#504
openshift-merge-bot[bot] merged 1 commit into
mainfrom
renovate/main-github-actions-dependencies

Conversation

@ospk8s-renovate

@ospk8s-renovate ospk8s-renovate Bot commented Sep 5, 2026

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Type Update Change
actions/deploy-pages (changelog) action digest cd2ce8f368f825

Configuration

📅 Schedule: (in timezone America/New_York)

  • Branch creation
    • "every weekend"
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about this update again.


  • If you want to rebase/retry this PR, check this box

This PR has been generated by Mend Renovate CLI.

@openshift-ci
openshift-ci Bot requested review from imatza-rh and lpiwowar September 5, 2026 05:45
@openshift-ci

openshift-ci Bot commented Sep 5, 2026

Copy link
Copy Markdown

Hi @ospk8s-renovate[bot]. Thanks for your PR.

I'm waiting for a openstack-k8s-operators member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work.

Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

@kstrenkova

Copy link
Copy Markdown
Contributor

/ok-to-test

@openshift-ci

openshift-ci Bot commented Sep 7, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: kstrenkova, ospk8s-renovate[bot]

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci openshift-ci Bot added the approved label Sep 7, 2026
@openshift-merge-bot
openshift-merge-bot Bot merged commit 5e3ed8c into main Sep 7, 2026
7 of 9 checks passed
@openshift-merge-bot
openshift-merge-bot Bot deleted the renovate/main-github-actions-dependencies branch September 7, 2026 07:06
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant