Skip to content

NO-ISSUE: rebase-release-5.0-5.0.0-0.nightly-2026-09-13-163645_amd64-2026-09-13_arm64-2026-09-13 - #7374

Closed
microshift-rebase-script[bot] wants to merge 12 commits into
release-5.0from
rebase-release-5.0-5.0.0-0.nightly-2026-09-13-163645_amd64-2026-09-13_arm64-2026-09-13
Closed

microshift-rebase-script[bot] wants to merge 12 commits into
release-5.0from
rebase-release-5.0-5.0.0-0.nightly-2026-09-13-163645_amd64-2026-09-13_arm64-2026-09-13

Conversation

@microshift-rebase-script

Copy link
Copy Markdown
Contributor

amd64: 5.0.0-0.nightly-2026-09-13-163645
arm64: 5.0.0-0.nightly-arm64-2026-09-13-223629
prow job: https://prow.ci.openshift.org/view/gs/origin-ci-test/logs/periodic-ci-openshift-microshift-release-5.0-periodics-rebase-on-nightlies/2099347777262718976

  • cluster-kube-apiserver-operator embedded-component 440392249bc58324156ba813b797728e9a81fa3c to b8c26db7c7c7e49476235ec83b2e81ef0996e242

    • 4b021f3 2026-09-11T16:11:55Z OCPBUGS-114428: Update build-machinery-go vendor dependency
  • cluster-kube-controller-manager-operator embedded-component 7be80f36a759d713170588c44fcd5b45970382d5 to 1ee372e7019b41fbcb4c64323142c4a06bf266d0

  • cluster-kube-scheduler-operator embedded-component 7b737a3f68ff52ebff3c9f541edbca357dde7111 to e0d271fceb3727138177e017f860a8ff79738bb3

    • a346b0f 2026-09-11T16:47:12Z OCPBUGS-122387: Update build-machinery-go vendor dependency
  • cluster-network-operator embedded-component 1b9a053663d46a789d8c500314cdfae3f879c060 to d7a55c9a073db7e954d3dd484e9d400dc5b5b3ac

    • 77d0804 2026-08-22T02:42:55+05:30 CVE-2026-41178: bump go.opentelemetry.io/otel to v1.44.0
  • kubernetes embedded-component 306002d4cb7d5621b4b9d1871d1d029682f8f824 to b376ee7b841bed5a51cb9520b37d87931c48a821

    • b9c17c8f6 2026-09-07T09:09:13+01:00 UPSTREAM: : bump otel to v1.44.0 to fix CVE-2026-41178
  • machine-config-operator embedded-component 3400a9c1429cce50d3a097f11e1a770e028e30d4 to 966718a5dab28fc1594d233a1293f2edc56620c9

    • 8d3cc3ff 2026-09-11T10:05:18Z updateLayeredOS deploy-from-self when skopeo < 1.22.2
    • 8422372e 2026-09-08T08:28:52Z OCPBUGS-116490: remove nft chains before checking the ignition config
    • c74b862a 2026-09-07T16:20:11Z operator: retry on conflict in syncMachineConfigNodes
    • 84ae2810 2026-08-27T18:04:39Z crio: drop container restore support
  • kubernetes image-amd64 306002d4cb7d5621b4b9d1871d1d029682f8f824 to b376ee7b841bed5a51cb9520b37d87931c48a821

    • b9c17c8f6 2026-09-07T09:09:13+01:00 UPSTREAM: : bump otel to v1.44.0 to fix CVE-2026-41178
  • kubernetes image-arm64 306002d4cb7d5621b4b9d1871d1d029682f8f824 to b376ee7b841bed5a51cb9520b37d87931c48a821

    • b9c17c8f6 2026-09-07T09:09:13+01:00 UPSTREAM: : bump otel to v1.44.0 to fix CVE-2026-41178

/label tide/merge-method-squash
/label backport-risk-assessed
/label jira/valid-bug

@microshift-rebase-script

Copy link
Copy Markdown
Contributor Author

/pipeline auto

@openshift-ci-robot openshift-ci-robot added the jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. label Sep 14, 2026
@openshift-ci-robot

Copy link
Copy Markdown

@microshift-rebase-script[bot]: This pull request explicitly references no jira issue.

Details

In response to this:

amd64: 5.0.0-0.nightly-2026-09-13-163645
arm64: 5.0.0-0.nightly-arm64-2026-09-13-223629
prow job: https://prow.ci.openshift.org/view/gs/origin-ci-test/logs/periodic-ci-openshift-microshift-release-5.0-periodics-rebase-on-nightlies/2099347777262718976

  • cluster-kube-apiserver-operator embedded-component 440392249bc58324156ba813b797728e9a81fa3c to b8c26db7c7c7e49476235ec83b2e81ef0996e242

  • 4b021f3 2026-09-11T16:11:55Z OCPBUGS-114428: Update build-machinery-go vendor dependency

  • cluster-kube-controller-manager-operator embedded-component 7be80f36a759d713170588c44fcd5b45970382d5 to 1ee372e7019b41fbcb4c64323142c4a06bf266d0

  • 0fedaa9 2026-09-11T16:18:27Z OCPBUGS-122387: Update build-machinery-go vendor dependency

  • a8b375f 2026-08-18T06:29:39Z Updating ose-cluster-kube-controller-manager-operator-container image to be consistent with ART for 5.0 Reconciling with https://github.com/openshift-eng/ocp-build-data/tree/ab3769c912df87c39f1bb9109164abd058bc1b52/images/ose-cluster-kube-controller-manager-operator.yml

  • cluster-kube-scheduler-operator embedded-component 7b737a3f68ff52ebff3c9f541edbca357dde7111 to e0d271fceb3727138177e017f860a8ff79738bb3

  • a346b0f 2026-09-11T16:47:12Z OCPBUGS-122387: Update build-machinery-go vendor dependency

  • cluster-network-operator embedded-component 1b9a053663d46a789d8c500314cdfae3f879c060 to d7a55c9a073db7e954d3dd484e9d400dc5b5b3ac

  • 77d0804 2026-08-22T02:42:55+05:30 CVE-2026-41178: bump go.opentelemetry.io/otel to v1.44.0

  • kubernetes embedded-component 306002d4cb7d5621b4b9d1871d1d029682f8f824 to b376ee7b841bed5a51cb9520b37d87931c48a821

  • b9c17c8f6 2026-09-07T09:09:13+01:00 UPSTREAM: : bump otel to v1.44.0 to fix CVE-2026-41178

  • machine-config-operator embedded-component 3400a9c1429cce50d3a097f11e1a770e028e30d4 to 966718a5dab28fc1594d233a1293f2edc56620c9

  • 8d3cc3ff 2026-09-11T10:05:18Z updateLayeredOS deploy-from-self when skopeo < 1.22.2

  • 8422372e 2026-09-08T08:28:52Z OCPBUGS-116490: remove nft chains before checking the ignition config

  • c74b862a 2026-09-07T16:20:11Z operator: retry on conflict in syncMachineConfigNodes

  • 84ae2810 2026-08-27T18:04:39Z crio: drop container restore support

  • kubernetes image-amd64 306002d4cb7d5621b4b9d1871d1d029682f8f824 to b376ee7b841bed5a51cb9520b37d87931c48a821

  • b9c17c8f6 2026-09-07T09:09:13+01:00 UPSTREAM: : bump otel to v1.44.0 to fix CVE-2026-41178

  • kubernetes image-arm64 306002d4cb7d5621b4b9d1871d1d029682f8f824 to b376ee7b841bed5a51cb9520b37d87931c48a821

  • b9c17c8f6 2026-09-07T09:09:13+01:00 UPSTREAM: : bump otel to v1.44.0 to fix CVE-2026-41178

/label tide/merge-method-squash
/label backport-risk-assessed
/label jira/valid-bug

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci

openshift-ci Bot commented Sep 14, 2026

Copy link
Copy Markdown
Contributor

Pipeline controller notification
This repo is configured to use the pipeline controller. Second-stage tests will be triggered either automatically or after lgtm label is added, depending on the repository configuration. The pipeline controller will automatically detect which contexts are required and will utilize /test Prow commands to trigger the second stage.

For optional jobs, comment /test ? to see a list of all defined jobs. To trigger manually all jobs from second stage use /pipeline required command.

This repository is configured in: LGTM mode

@coderabbitai

coderabbitai Bot commented Sep 14, 2026

Copy link
Copy Markdown
Contributor

Important

Review skipped

Bot user detected.

To trigger a single review, invoke the @coderabbitai review command.

⚙️ Run configuration

Configuration used: Repository YAML (base), Central YAML (inherited)

Review profile: CHILL

Plan: Advanced

Run ID: 49c4c5c2-0d8c-4bc4-a625-07b46a247c9f

You can disable this status message by setting the reviews.review_status to false in the CodeRabbit configuration file.

Use the checkbox below for a quick retry:

  • 🔍 Trigger review

Comment @coderabbitai help to get the list of available commands.

@openshift-ci openshift-ci Bot added tide/merge-method-squash Denotes a PR that should be squashed by tide when it merges. backport-risk-assessed Indicates a PR to a release branch has been evaluated and considered safe to accept. jira/valid-bug Indicates that a referenced Jira bug is valid for the branch this PR is targeting. pipeline-auto labels Sep 14, 2026
@openshift-ci

openshift-ci Bot commented Sep 14, 2026

Copy link
Copy Markdown
Contributor

Pipeline controller notification

The pipeline-auto label has been added to this PR. Second-stage tests will be triggered automatically when all first-stage tests pass.

@openshift-ci
openshift-ci Bot requested review from pacevedom and pmtk September 14, 2026 04:40
@openshift-ci

openshift-ci Bot commented Sep 14, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: microshift-rebase-script[bot]
Once this PR has been reviewed and has the lgtm label, please assign eslutsky for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci

openshift-ci Bot commented Sep 14, 2026

Copy link
Copy Markdown
Contributor

@microshift-rebase-script[bot]: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@openshift-ci

openshift-ci Bot commented Sep 14, 2026

Copy link
Copy Markdown
Contributor

Scheduling tests matching the pipeline_run_if_changed or not excluded by pipeline_skip_if_only_changed parameters:
/test e2e-aws-tests
/test e2e-aws-tests-arm
/test e2e-aws-tests-bootc-arm-el10
/test e2e-aws-tests-bootc-arm-el9
/test e2e-aws-tests-bootc-el10
/test e2e-aws-tests-bootc-el9

1 similar comment
@openshift-ci

openshift-ci Bot commented Sep 14, 2026

Copy link
Copy Markdown
Contributor

Scheduling tests matching the pipeline_run_if_changed or not excluded by pipeline_skip_if_only_changed parameters:
/test e2e-aws-tests
/test e2e-aws-tests-arm
/test e2e-aws-tests-bootc-arm-el10
/test e2e-aws-tests-bootc-arm-el9
/test e2e-aws-tests-bootc-el10
/test e2e-aws-tests-bootc-el9

@microshift-rebase-script

Copy link
Copy Markdown
Contributor Author

Closing as duplicate: superseded by #7380.
/close
Added by prow-jobs-for-pull-requests.sh 🤖

@openshift-ci openshift-ci Bot closed this Sep 15, 2026
@openshift-ci

openshift-ci Bot commented Sep 15, 2026

Copy link
Copy Markdown
Contributor

@microshift-rebase-script[bot]: Closed this PR.

Details

In response to this:

Closing as duplicate: superseded by #7380.
/close
Added by prow-jobs-for-pull-requests.sh 🤖

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

backport-risk-assessed Indicates a PR to a release branch has been evaluated and considered safe to accept. jira/valid-bug Indicates that a referenced Jira bug is valid for the branch this PR is targeting. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. pipeline-auto tide/merge-method-squash Denotes a PR that should be squashed by tide when it merges.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant