build(deps): bump the aws-sdk group with 5 updates - #1006
dependabot[bot] wants to merge 1 commit into
Conversation
Bumps the aws-sdk group with 5 updates: | Package | From | To | | --- | --- | --- | | [github.com/aws/aws-sdk-go-v2](https://github.com/aws/aws-sdk-go-v2) | `1.47.0` | `1.47.1` | | [github.com/aws/aws-sdk-go-v2/config](https://github.com/aws/aws-sdk-go-v2) | `1.33.5` | `1.33.6` | | [github.com/aws/aws-sdk-go-v2/credentials](https://github.com/aws/aws-sdk-go-v2) | `1.20.5` | `1.20.6` | | [github.com/aws/aws-sdk-go-v2/service/sts](https://github.com/aws/aws-sdk-go-v2) | `1.51.0` | `1.51.1` | | [github.com/aws/aws-sdk-go-v2/service/ssm](https://github.com/aws/aws-sdk-go-v2) | `1.78.0` | `1.78.1` | Updates `github.com/aws/aws-sdk-go-v2` from 1.47.0 to 1.47.1 - [Release notes](https://github.com/aws/aws-sdk-go-v2/releases) - [Commits](aws/aws-sdk-go-v2@v1.47.0...v1.47.1) Updates `github.com/aws/aws-sdk-go-v2/config` from 1.33.5 to 1.33.6 - [Release notes](https://github.com/aws/aws-sdk-go-v2/releases) - [Commits](aws/aws-sdk-go-v2@config/v1.33.5...config/v1.33.6) Updates `github.com/aws/aws-sdk-go-v2/credentials` from 1.20.5 to 1.20.6 - [Release notes](https://github.com/aws/aws-sdk-go-v2/releases) - [Commits](aws/aws-sdk-go-v2@service/mq/v1.20.5...service/mq/v1.20.6) Updates `github.com/aws/aws-sdk-go-v2/service/sts` from 1.51.0 to 1.51.1 - [Release notes](https://github.com/aws/aws-sdk-go-v2/releases) - [Commits](aws/aws-sdk-go-v2@service/s3/v1.51.0...service/s3/v1.51.1) Updates `github.com/aws/aws-sdk-go-v2/service/ssm` from 1.78.0 to 1.78.1 - [Release notes](https://github.com/aws/aws-sdk-go-v2/releases) - [Commits](aws/aws-sdk-go-v2@service/s3/v1.78.0...service/s3/v1.78.1) --- updated-dependencies: - dependency-name: github.com/aws/aws-sdk-go-v2 dependency-version: 1.47.1 dependency-type: direct:production update-type: version-update:semver-patch dependency-group: aws-sdk - dependency-name: github.com/aws/aws-sdk-go-v2/config dependency-version: 1.33.6 dependency-type: direct:production update-type: version-update:semver-patch dependency-group: aws-sdk - dependency-name: github.com/aws/aws-sdk-go-v2/credentials dependency-version: 1.20.6 dependency-type: direct:production update-type: version-update:semver-patch dependency-group: aws-sdk - dependency-name: github.com/aws/aws-sdk-go-v2/service/sts dependency-version: 1.51.1 dependency-type: direct:production update-type: version-update:semver-patch dependency-group: aws-sdk - dependency-name: github.com/aws/aws-sdk-go-v2/service/ssm dependency-version: 1.78.1 dependency-type: direct:production update-type: version-update:semver-patch dependency-group: aws-sdk ... Signed-off-by: dependabot[bot] <support@github.com>
|
Important Review skippedIgnore keyword(s) in the title. ⛔ Ignored keywords (3)
Please check the settings in the CodeRabbit UI or the ⚙️ Run configurationConfiguration used: Repository: openshift/backplane-cli/.coderabbit.yaml Review profile: CHILL Plan: Enterprise Run ID: You can disable this status message by setting the Use the checkbox below for a quick retry:
Comment |
|
Hi @dependabot[bot]. Thanks for your PR. I'm waiting for a openshift member to verify that this patch is reasonable to test. If it is, they should reply with Regular contributors should join the org to skip this step. Once the patch is verified, the new status will be reflected by the I understand the commands that are listed here. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: dependabot[bot] The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
Codecov Report✅ All modified and coverable lines are covered by tests. Additional details and impacted files@@ Coverage Diff @@
## main #1006 +/- ##
=======================================
Coverage 55.86% 55.86%
=======================================
Files 78 78
Lines 6159 6159
=======================================
Hits 3441 3441
Misses 2311 2311
Partials 407 407 🚀 New features to boost your workflow:
|
Bumps the aws-sdk group with 5 updates:
1.47.01.47.11.33.51.33.61.20.51.20.61.51.01.51.11.78.01.78.1Updates
github.com/aws/aws-sdk-go-v2from 1.47.0 to 1.47.1Commits
52ba256Release 2026-09-2421223cdRegenerated Clients84d89bdUpdate endpoints modeld9c59ffUpdate API model8fd86b1feat: Add aws-chunked response decoding reader (#3563)ff3ac6fRelease 2026-09-23e57d2ffRegenerated Clientsef6fdb9Update endpoints model385ab15Update API modelfafcb4fChecksum non-200 HTTP responses (#3564)Updates
github.com/aws/aws-sdk-go-v2/configfrom 1.33.5 to 1.33.6Commits
52ba256Release 2026-09-2421223cdRegenerated Clients84d89bdUpdate endpoints modeld9c59ffUpdate API model8fd86b1feat: Add aws-chunked response decoding reader (#3563)ff3ac6fRelease 2026-09-23e57d2ffRegenerated Clientsef6fdb9Update endpoints model385ab15Update API modelfafcb4fChecksum non-200 HTTP responses (#3564)Updates
github.com/aws/aws-sdk-go-v2/credentialsfrom 1.20.5 to 1.20.6Commits
27acb6bRelease 2023-12-15d3d958fRegenerated Clientsb9e42b1Update endpoints modeld61965fUpdate API modeldca6a24Release 2023-12-1465183edRegenerated Clients9a01654Update endpoints model87eb076Update API modelaaea7b5Release 2023-12-138b61d81Regenerated ClientsUpdates
github.com/aws/aws-sdk-go-v2/service/stsfrom 1.51.0 to 1.51.1Commits
7095341Release 2024-02-23bcf04e6Regenerated Clientsca190b0Update API model6397a64move common middleware stack ops to service client modules (#2516)Updates
github.com/aws/aws-sdk-go-v2/service/ssmfrom 1.78.0 to 1.78.1Commits
9bd153cRelease 2025-03-04.2b967446Regenerated Clients188c52aUpdate endpoints modeldefaf5cUpdate API model8f2dd23add test that verifies SRA order of operations (#3025)3d547b0Release 2025-03-04554a149Regenerated Clients3abb221Update endpoints model95ae39dUpdate API modelc62ec38create kitchen sink test service (#3023)Dependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting
@dependabot rebase.Dependabot commands and options
You can trigger Dependabot actions by commenting on this PR:
@dependabot rebasewill rebase this PR@dependabot recreatewill recreate this PR, overwriting any edits that have been made to it@dependabot show <dependency name> ignore conditionswill show all of the ignore conditions of the specified dependency@dependabot ignore <dependency name> major versionwill close this group update PR and stop Dependabot creating any more for the specific dependency's major version (unless you unignore this specific dependency's major version or upgrade to it yourself)@dependabot ignore <dependency name> minor versionwill close this group update PR and stop Dependabot creating any more for the specific dependency's minor version (unless you unignore this specific dependency's minor version or upgrade to it yourself)@dependabot ignore <dependency name>will close this group update PR and stop Dependabot creating any more for the specific dependency (unless you unignore this specific dependency or upgrade to it yourself)@dependabot unignore <dependency name>will remove all of the ignore conditions of the specified dependency@dependabot unignore <dependency name> <ignore condition>will remove the ignore condition of the specified dependency and ignore conditions