Skip to content
This repository was archived by the owner on Jul 17, 2026. It is now read-only.

Set maintainers as CODEOWNERS - #540

Merged
gonzolino merged 2 commits into
ironcore-dev:mainfrom
gonzolino:chore/codeowners
Jul 13, 2026
Merged

Set maintainers as CODEOWNERS#540
gonzolino merged 2 commits into
ironcore-dev:mainfrom
gonzolino:chore/codeowners

Conversation

@gonzolino

Copy link
Copy Markdown
Contributor

Add the maintainer team(s) for this repo to the CODEOWNERS file.

Signed-off-by: Daniel Gonzalez Nothnagel <daniel.gonzalez.nothnagel@sap.com>
Signed-off-by: Daniel Gonzalez Nothnagel <daniel.gonzalez.nothnagel@sap.com>
@gonzolino
gonzolino requested a review from a team July 13, 2026 13:19
@gonzolino
gonzolino merged commit 6e9525f into ironcore-dev:main Jul 13, 2026
10 of 14 checks passed
@gonzolino
gonzolino deleted the chore/codeowners branch July 13, 2026 15:16
@hardikdr hardikdr added this to Roadmap Aug 22, 2026
@github-project-automation github-project-automation Bot moved this to Done in Roadmap Aug 22, 2026
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

Projects

Status: Done

Development

Successfully merging this pull request may close these issues.

2 participants