Skip to content

Add clustersim command for manual MCM testing - #1152

Open
takoverflow wants to merge 2 commits into
gardener:masterfrom
takoverflow:feat/clustersim
Open

takoverflow wants to merge 2 commits into
gardener:masterfrom
takoverflow:feat/clustersim

Conversation

@takoverflow

@takoverflow takoverflow commented Sep 11, 2026

Copy link
Copy Markdown
Member

What this PR does / why we need it:
Introduces standalone clustersim tool to create virtual clusters with workload consisting of MachineClasses and MachineDeployments to aid in manual testing. Please check its README.md for more details.

Which issue(s) this PR fixes:
Fixes part of #1120

Special notes for your reviewer:
The line diff is bloated because of the instances.json file, which consists of capacity information for various AWS instances. If deemed overkill, it can be replaced by a list of top 20 frequently used instances for all providers.

/invite @elankath

Release note:

Added a command line program `clustersim` to aid in manual MCM testing via virtual clusters

Signed-off-by: Prashant Tak <prashant.tak@sap.com>
@gardener-prow gardener-prow Bot added the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Sep 11, 2026
@gardener-prow

gardener-prow Bot commented Sep 11, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign ashwani2k for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@gardener-prow gardener-prow Bot added do-not-merge/needs-kind Indicates a PR lacks a `kind/foo` label and requires one. size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files. labels Sep 11, 2026
@takoverflow takoverflow added kind/feature and removed do-not-merge/needs-kind Indicates a PR lacks a `kind/foo` label and requires one. labels Sep 11, 2026
@takoverflow
takoverflow marked this pull request as ready for review September 15, 2026 19:54
@takoverflow
takoverflow requested a review from a team as a code owner September 15, 2026 19:54
@gardener-prow gardener-prow Bot removed the do-not-merge/work-in-progress Indicates that a PR should not merge because it is a work in progress. label Sep 15, 2026
Signed-off-by: Prashant Tak <prashant.tak@sap.com>
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

kind/feature size/XXL Denotes a PR that changes 1000+ lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant