Conversation
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
6 tasks
thiyyakat
reviewed
Aug 7, 2026
thiyyakat
left a comment
Member
There was a problem hiding this comment.
Thanks for the changes! The test looks good to me. I just had a few questions about possible behaviour.
gagan16k
reviewed
Aug 28, 2026
takoverflow
reviewed
Sep 8, 2026
aaronfern
force-pushed
the
node_del_it
branch
from
September 10, 2026 05:22
f331ce4 to
86ada96
Compare
takoverflow
reviewed
Sep 11, 2026
Signed-off-by: aaronfern <aaron.francis.fernandes@sap.com>
aaronfern
force-pushed
the
node_del_it
branch
from
September 11, 2026 10:15
86ada96 to
a035311
Compare
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
What this PR does / why we need it:
This PR add an integration test to verify if machine deletion happens when node deletion is triggered
Which issue(s) this PR fixes:
Fixes partially #1123
Special notes for your reviewer:
Release note: