Skip to content

fix(nav): Remove upgrades & redo home - #144

Merged
mbeutt-cb merged 7 commits into
mainfrom
remove-upgrades
Sep 8, 2026
Merged

fix(nav): Remove upgrades & redo home#144
mbeutt-cb merged 7 commits into
mainfrom
remove-upgrades

Conversation

@mbeutt-cb

Copy link
Copy Markdown
Collaborator

Summary

  • Removed upgrades and changelog now that they are on the docs site
  • Added more content to the homepage to give it more life and purpose while we add additional features to the site

Screenshots

home

Test plan

  • Ran npm run test locally and audited all impacted pages

@cb-heimdall

cb-heimdall commented Sep 3, 2026

Copy link
Copy Markdown
Collaborator

✅ Heimdall Review Status

Requirement Status More Info
Reviews 1/1
Denominator calculation
Show calculation
1 if user is bot 0
1 if user is external 0
2 if repo is sensitive 0
From .codeflow.yml 1
Additional review requirements
Show calculation
Max 0
0
From CODEOWNERS 0
Global minimum 0
Max 1
1
1 if commit is unverified 0
Sum 1

@vercel

vercel Bot commented Sep 3, 2026

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

Project Deployment Actions Updated
omni-ui Ready Ready Preview Sep 8, 2026 1:32pm UTC

Request Review

@mbeutt-cb
mbeutt-cb added this pull request to the merge queue Sep 8, 2026
Merged via the queue into main with commit cebee0d Sep 8, 2026
13 checks passed
@mbeutt-cb
mbeutt-cb deleted the remove-upgrades branch September 8, 2026 15:49
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants