Skip to content

Bump github.com/google/cel-go from 0.26.0 to 0.29.0 - #2080

Merged
openshift-merge-bot[bot] merged 1 commit into
mainfrom
dependabot/go_modules/github.com/google/cel-go-0.29.0
Sep 10, 2026
Merged

openshift-merge-bot[bot] merged 1 commit into
mainfrom
dependabot/go_modules/github.com/google/cel-go-0.29.0

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Jul 28, 2026

Copy link
Copy Markdown
Contributor

Bumps github.com/google/cel-go from 0.26.0 to 0.29.0.

Commits

@dependabot dependabot Bot added dependencies Pull requests that update a dependency file go Pull requests that update Go code labels Jul 28, 2026
@openshift-ci openshift-ci Bot added the size/S label Jul 28, 2026
@sonarqubecloud

Copy link
Copy Markdown

@tesshuflower

Copy link
Copy Markdown
Contributor

@dependabot rebase

Bumps [github.com/google/cel-go](https://github.com/google/cel-go) from 0.26.0 to 0.29.0.
- [Commits](https://github.com/google/cel-go/commits)

---
updated-dependencies:
- dependency-name: github.com/google/cel-go
  dependency-version: 0.29.0
  dependency-type: indirect
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot
dependabot Bot force-pushed the dependabot/go_modules/github.com/google/cel-go-0.29.0 branch from 65b9857 to 3e3e9be Compare September 10, 2026 16:11
@sonarqubecloud

Copy link
Copy Markdown

@tesshuflower

Copy link
Copy Markdown
Contributor

/approve
/lgtm

@openshift-ci

openshift-ci Bot commented Sep 10, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: dependabot[bot], tesshuflower

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-merge-bot
openshift-merge-bot Bot merged commit a604d56 into main Sep 10, 2026
36 checks passed
@dependabot
dependabot Bot deleted the dependabot/go_modules/github.com/google/cel-go-0.29.0 branch September 10, 2026 19:55
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved dependencies Pull requests that update a dependency file go Pull requests that update Go code lgtm size/S

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant