Update dependency kubernetes-csi/csi-driver-host-path to v1.18.0 - #2073
renovate-bot wants to merge 1 commit into
Conversation
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: renovate-bot The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
@renovate-bot: The following test failed, say
Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here. |
Signed-off-by: Mend Renovate <bot@renovateapp.com>
b73367e to
88d8b75
Compare
|
Codecov Report✅ All modified and coverable lines are covered by tests. Additional details and impacted files@@ Coverage Diff @@
## main #2073 +/- ##
=======================================
- Coverage 68.8% 68.7% -0.1%
=======================================
Files 58 58
Lines 6076 6076
=======================================
- Hits 4184 4178 -6
- Misses 1582 1585 +3
- Partials 310 313 +3 🚀 New features to boost your workflow:
|



This PR contains the following updates:
v1.17.1→v1.18.0Release Notes
kubernetes-csi/csi-driver-host-path (kubernetes-csi/csi-driver-host-path)
v1.18.0Compare Source
Container
docker pull registry.k8s.io/sig-storage/hostpathplugin:v1.18.0
Full Changelog
https://github.com/kubernetes-csi/csi-driver-host-path/blob/v1.18.0/CHANGELOG/CHANGELOG-1.18.md
Configuration
📅 Schedule: (in timezone America/New_York)
* 3-6 * * 1-5)🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.
♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.
🔕 Ignore: Close this PR and you won't be reminded about this update again.
This PR was generated by Mend Renovate. View the repository job log.