Skip to content

Reduce GKE test runtimes and quota pressure - #18982

Merged
rileykarson merged 7 commits into
GoogleCloudPlatform:mainfrom
rileykarson:gke-compute
Sep 14, 2026
Merged

rileykarson merged 7 commits into
GoogleCloudPlatform:mainfrom
rileykarson:gke-compute

Conversation

@rileykarson

Copy link
Copy Markdown
Member
  • Fix sweeper to handle more locations, add guards
  • Reduce the VM size and node count where possible

Release Note Template for Downstream PRs (will be copied)

See Write release notes for guidance.

@modular-magician

modular-magician commented Sep 12, 2026

Copy link
Copy Markdown
Collaborator

Hi there, I'm the Modular magician. I've detected the following information about your changes for commit b6e4445:

Diff report

Your PR generated the following diffs in downstream repositories:

Repository Diff Link Changes
google provider View Diff 3 files changed, 161 insertions(+), 78 deletions(-)
google-beta provider View Diff 3 files changed, 162 insertions(+), 79 deletions(-)

Test report

Analytics

Total Tests Passed Skipped Affected
323 258 17 48
Affected Service Packages
  • container

Learn how VCR tests work


Step 1: Replaying Mode

Action taken

Found 48 affected test(s) by replaying old test recordings. Starting RECORDING based on the most recent commit.

Click here to see the affected tests
  • TestAccContainerCluster_autopilot_privileged_admission
  • TestAccContainerCluster_backend
  • TestAccContainerCluster_dataplaneOptimizationMode
  • TestAccContainerCluster_disableL4LbFirewallReconciliation
  • TestAccContainerCluster_misc
  • TestAccContainerCluster_network_tier_config
  • TestAccContainerCluster_omittedILBSubsetting
  • TestAccContainerCluster_skipNodePoolRefresh
  • TestAccContainerCluster_withAutopilotClusterPolicy
  • TestAccContainerCluster_withClusterBootDisk
  • TestAccContainerCluster_withFilteredNotificationConfig
  • TestAccContainerCluster_withFlexiblePodCIDR
  • TestAccContainerCluster_withILBSubsetting
  • TestAccContainerCluster_withMasterAuthConfig_NoCert
  • TestAccContainerCluster_withMeshCertificatesConfig
  • TestAccContainerCluster_withNetworkPolicyEnabled
  • TestAccContainerCluster_withNodeConfigScopeAlias
  • TestAccContainerCluster_withNodeConfigShieldedInstanceConfig
  • TestAccContainerCluster_withNodePoolAutoscaling
  • TestAccContainerCluster_withNodePoolBasic
  • TestAccContainerCluster_withNodePoolCIA
  • TestAccContainerCluster_withNodePoolUpdateVersion
  • TestAccContainerCluster_withNotificationConfig
  • TestAccContainerCluster_withWorkloadIdentityConfig
  • TestAccContainerNodePool_EmptyGuestAccelerator
  • TestAccContainerNodePool_acceleratorNetworkProfile
  • TestAccContainerNodePool_autoscaling
  • TestAccContainerNodePool_basic
  • TestAccContainerNodePool_basicWithClusterId
  • TestAccContainerNodePool_concurrent
  • TestAccContainerNodePool_defaultDriverInstallation
  • TestAccContainerNodePool_ignoreNodeCountChanges
  • TestAccContainerNodePool_maxPodsPerNode
  • TestAccContainerNodePool_regionalAutoscaling
  • TestAccContainerNodePool_regionalClusters
  • TestAccContainerNodePool_resourceManagerTags
  • TestAccContainerNodePool_shieldedInstanceConfig
  • TestAccContainerNodePool_totalSize
  • TestAccContainerNodePool_withLegacyNodeConfigAndBootDiskUpdate
  • TestAccContainerNodePool_withMachineAndDiskUpdate
  • TestAccContainerNodePool_withMachineDiskStoragePoolsUpdate
  • TestAccContainerNodePool_withManagement
  • TestAccContainerNodePool_withNodeConfigWithBootDiskConfig
  • TestAccContainerNodePool_withNodeConfigWithBootDiskConfigChangeType
  • TestAccContainerNodePool_withNodeConfigWithBootDiskConfigChangeTypeLegacy
  • TestAccContainerNodePool_withRDMA
  • TestAccContainerNodePool_withSoleTenantConfig
  • TestAccContainerNodePool_withTaintsUpdate

View the replaying VCR build log


Step 2: Recording Mode

Recording Mode Replaying Rerun Test Name
✅ Log TestAccContainerCluster_backend
✅ Log TestAccContainerCluster_disableL4LbFirewallReconciliation
✅ Log TestAccContainerCluster_network_tier_config
✅ Log TestAccContainerCluster_omittedILBSubsetting
✅ Log TestAccContainerCluster_withClusterBootDisk
✅ Log TestAccContainerCluster_withFilteredNotificationConfig
✅ Log TestAccContainerCluster_withFlexiblePodCIDR
✅ Log TestAccContainerCluster_withILBSubsetting
✅ Log TestAccContainerCluster_withMasterAuthConfig_NoCert
✅ Log TestAccContainerCluster_withNodeConfigScopeAlias
✅ Log TestAccContainerCluster_withNodeConfigShieldedInstanceConfig
✅ Log TestAccContainerCluster_withNodePoolAutoscaling
✅ Log TestAccContainerCluster_withNodePoolBasic
✅ Log TestAccContainerCluster_withNodePoolCIA
✅ Log TestAccContainerCluster_withNodePoolUpdateVersion
✅ Log TestAccContainerCluster_withNotificationConfig
✅ Log TestAccContainerNodePool_EmptyGuestAccelerator
✅ Log TestAccContainerNodePool_basicWithClusterId
✅ Log TestAccContainerNodePool_defaultDriverInstallation
✅ Log TestAccContainerNodePool_maxPodsPerNode
✅ Log TestAccContainerNodePool_regionalAutoscaling
✅ Log TestAccContainerNodePool_regionalClusters
✅ Log TestAccContainerNodePool_shieldedInstanceConfig
✅ Log TestAccContainerNodePool_totalSize
✅ Log TestAccContainerNodePool_withLegacyNodeConfigAndBootDiskUpdate
✅ Log TestAccContainerNodePool_withMachineAndDiskUpdate
✅ Log TestAccContainerNodePool_withMachineDiskStoragePoolsUpdate
✅ Log TestAccContainerNodePool_withManagement
✅ Log TestAccContainerNodePool_withNodeConfigWithBootDiskConfig
✅ Log TestAccContainerNodePool_withNodeConfigWithBootDiskConfigChangeType
✅ Log TestAccContainerNodePool_withNodeConfigWithBootDiskConfigChangeTypeLegacy
✅ Log TestAccContainerNodePool_withTaintsUpdate
❌ Error · Log - TestAccContainerCluster_autopilot_privileged_admission
❌ Error · Log - TestAccContainerCluster_dataplaneOptimizationMode
❌ Error · Log - TestAccContainerCluster_misc
❌ Error · Log - TestAccContainerCluster_skipNodePoolRefresh
❌ Error · Log - TestAccContainerCluster_withAutopilotClusterPolicy
❌ Error · Log - TestAccContainerCluster_withMeshCertificatesConfig
❌ Error · Log - TestAccContainerCluster_withNetworkPolicyEnabled
❌ Error · Log - TestAccContainerCluster_withWorkloadIdentityConfig
❌ Error · Log - TestAccContainerNodePool_acceleratorNetworkProfile
❌ Error · Log - TestAccContainerNodePool_autoscaling
❌ Error · Log - TestAccContainerNodePool_basic
❌ Error · Log - TestAccContainerNodePool_concurrent
❌ Error · Log - TestAccContainerNodePool_ignoreNodeCountChanges
❌ Error · Log - TestAccContainerNodePool_resourceManagerTags
❌ Error · Log - TestAccContainerNodePool_withRDMA
❌ Error · Log - TestAccContainerNodePool_withSoleTenantConfig

Caution

Issues requiring attention before PR completion

🔴 Initial Recording Failed: Some tests failed during the recording step. See the table above for details.

Please address these issues to complete your PR. If you believe these detections are incorrect or unrelated to your change, please raise the concern with your reviewer.

View the recording VCR build log or the debug logs folder for detailed results.

@rileykarson VCR tests complete for b6e4445!

@modular-magician

modular-magician commented Sep 12, 2026

Copy link
Copy Markdown
Collaborator

Hi there, I'm the Modular magician. I've detected the following information about your changes for commit 5d5933b:

Diff report

Your PR generated the following diffs in downstream repositories:

Repository Diff Link Changes
google provider View Diff 3 files changed, 237 insertions(+), 78 deletions(-)
google-beta provider View Diff 3 files changed, 238 insertions(+), 79 deletions(-)

Test report

Analytics

Total Tests Passed Skipped Affected
323 290 17 16
Affected Service Packages
  • container

Learn how VCR tests work


Step 1: Replaying Mode

Action taken

Found 16 affected test(s) by replaying old test recordings. Starting RECORDING based on the most recent commit.

Click here to see the affected tests
  • TestAccContainerCluster_autopilot_privileged_admission
  • TestAccContainerCluster_dataplaneOptimizationMode
  • TestAccContainerCluster_misc
  • TestAccContainerCluster_skipNodePoolRefresh
  • TestAccContainerCluster_withAutopilotClusterPolicy
  • TestAccContainerCluster_withMeshCertificatesConfig
  • TestAccContainerCluster_withNetworkPolicyEnabled
  • TestAccContainerCluster_withWorkloadIdentityConfig
  • TestAccContainerNodePool_acceleratorNetworkProfile
  • TestAccContainerNodePool_autoscaling
  • TestAccContainerNodePool_basic
  • TestAccContainerNodePool_concurrent
  • TestAccContainerNodePool_ignoreNodeCountChanges
  • TestAccContainerNodePool_resourceManagerTags
  • TestAccContainerNodePool_withRDMA
  • TestAccContainerNodePool_withSoleTenantConfig

View the replaying VCR build log


Step 2: Recording Mode

Recording Mode Replaying Rerun Test Name
✅ Log TestAccContainerCluster_misc
✅ Log TestAccContainerCluster_skipNodePoolRefresh
✅ Log TestAccContainerCluster_withAutopilotClusterPolicy
✅ Log TestAccContainerCluster_withMeshCertificatesConfig
✅ Log TestAccContainerCluster_withNetworkPolicyEnabled
✅ Log TestAccContainerCluster_withWorkloadIdentityConfig
✅ Log TestAccContainerNodePool_autoscaling
✅ Log TestAccContainerNodePool_basic
✅ Log TestAccContainerNodePool_concurrent
✅ Log TestAccContainerNodePool_ignoreNodeCountChanges
✅ Log TestAccContainerNodePool_resourceManagerTags
❌ Error · Log - TestAccContainerCluster_autopilot_privileged_admission
❌ Error · Log - TestAccContainerCluster_dataplaneOptimizationMode
❌ Error · Log - TestAccContainerNodePool_acceleratorNetworkProfile
❌ Error · Log - TestAccContainerNodePool_withRDMA
❌ Error · Log - TestAccContainerNodePool_withSoleTenantConfig

Caution

Issues requiring attention before PR completion

🔴 Initial Recording Failed: Some tests failed during the recording step. See the table above for details.

Please address these issues to complete your PR. If you believe these detections are incorrect or unrelated to your change, please raise the concern with your reviewer.

View the recording VCR build log or the debug logs folder for detailed results.

@rileykarson VCR tests complete for 5d5933b!

@modular-magician

modular-magician commented Sep 12, 2026

Copy link
Copy Markdown
Collaborator

Hi there, I'm the Modular magician. I've detected the following information about your changes for commit 2cd7598:

Diff report

Your PR generated the following diffs in downstream repositories:

Repository Diff Link Changes
google provider View Diff 3 files changed, 237 insertions(+), 78 deletions(-)
google-beta provider View Diff 3 files changed, 238 insertions(+), 79 deletions(-)

Test report

Analytics

Total Tests Passed Skipped Affected
323 301 17 5
Affected Service Packages
  • container

Learn how VCR tests work


Step 1: Replaying Mode

Action taken

Found 5 affected test(s) by replaying old test recordings. Starting RECORDING based on the most recent commit.

Click here to see the affected tests
  • TestAccContainerCluster_autopilot_privileged_admission
  • TestAccContainerCluster_dataplaneOptimizationMode
  • TestAccContainerNodePool_acceleratorNetworkProfile
  • TestAccContainerNodePool_withRDMA
  • TestAccContainerNodePool_withSoleTenantConfig

View the replaying VCR build log


Step 2: Recording Mode

Recording Mode Replaying Rerun Test Name
✅ Log TestAccContainerNodePool_acceleratorNetworkProfile
✅ Log TestAccContainerNodePool_withRDMA
❌ Error · Log - TestAccContainerCluster_autopilot_privileged_admission
❌ Error · Log - TestAccContainerCluster_dataplaneOptimizationMode
❌ Error · Log - TestAccContainerNodePool_withSoleTenantConfig

Caution

Issues requiring attention before PR completion

🔴 Initial Recording Failed: Some tests failed during the recording step. See the table above for details.

Please address these issues to complete your PR. If you believe these detections are incorrect or unrelated to your change, please raise the concern with your reviewer.

View the recording VCR build log or the debug logs folder for detailed results.

@rileykarson VCR tests complete for 2cd7598!

@rileykarson
rileykarson requested review from SirGitsalot and c2thorn and removed request for c2thorn September 14, 2026 17:15

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

In most of the modified configs both google_container_cluster and google_container_node_pool had initial_node_count changed, others had only a google_container_cluster change. That changes some of the tests, which have a mixture of configs, e.g., TestAccContainerNodePool_totalSize has three steps:

testAccContainerNodePool_totalSize        cluster 3->1  np 2->2
testAccContainerNodePool_updateTotalSize  cluster 3->1  np 2->2
testAccContainerNodePool_basicTotalSize   cluster 3->1  np 2->1

This seems to affect TestAccContainerCluster_withNodePoolAutoscaling, TestAccContainerCluster_withNodePoolCIA, TestAccContainerNodePool_autoscaling, and TestAccContainerNodePool_regionalAutoscaling too.

Copy link
Copy Markdown
Member Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Good catch, thanks

I'll follow up and force explicit update checks in a separate PR

…er_cluster_sweeper.go

Co-authored-by: Chris Hawk <chrishawk+github@gmail.com>
@modular-magician

modular-magician commented Sep 14, 2026

Copy link
Copy Markdown
Collaborator

Hi there, I'm the Modular magician. I've detected the following information about your changes for commit 64bd005:

Diff report

Your PR generated the following diffs in downstream repositories:

Repository Diff Link Changes
google provider View Diff 3 files changed, 237 insertions(+), 78 deletions(-)
google-beta provider View Diff 3 files changed, 238 insertions(+), 79 deletions(-)

Test report

Analytics

Total Tests Passed Skipped Affected
323 303 17 3
Affected Service Packages
  • container

Learn how VCR tests work


Step 1: Replaying Mode

Action taken

Found 3 affected test(s) by replaying old test recordings. Starting RECORDING based on the most recent commit.

Click here to see the affected tests
  • TestAccContainerCluster_autopilot_privileged_admission
  • TestAccContainerCluster_dataplaneOptimizationMode
  • TestAccContainerNodePool_withSoleTenantConfig

View the replaying VCR build log


Step 2: Recording Mode

Recording Mode Replaying Rerun Test Name
❌ Error · Log - TestAccContainerCluster_autopilot_privileged_admission
❌ Error · Log - TestAccContainerCluster_dataplaneOptimizationMode
❌ Error · Log - TestAccContainerNodePool_withSoleTenantConfig

Caution

Issues requiring attention before PR completion

🔴 Initial Recording Failed: Some tests failed during the recording step. See the table above for details.

Please address these issues to complete your PR. If you believe these detections are incorrect or unrelated to your change, please raise the concern with your reviewer.

View the recording VCR build log or the debug logs folder for detailed results.

@rileykarson VCR tests complete for 64bd005!

@modular-magician

modular-magician commented Sep 14, 2026

Copy link
Copy Markdown
Collaborator

Hi there, I'm the Modular magician. I've detected the following information about your changes for commit dabe5a8:

Diff report

Your PR generated the following diffs in downstream repositories:

Repository Diff Link Changes
google provider View Diff 3 files changed, 240 insertions(+), 81 deletions(-)
google-beta provider View Diff 3 files changed, 241 insertions(+), 82 deletions(-)

Test report

Analytics

Total Tests Passed Skipped Affected
323 298 17 8
Affected Service Packages
  • container

Learn how VCR tests work


Step 1: Replaying Mode

Action taken

Found 8 affected test(s) by replaying old test recordings. Starting RECORDING based on the most recent commit.

Click here to see the affected tests
  • TestAccContainerCluster_autopilot_privileged_admission
  • TestAccContainerCluster_dataplaneOptimizationMode
  • TestAccContainerCluster_withNodePoolAutoscaling
  • TestAccContainerCluster_withNodePoolCIA
  • TestAccContainerNodePool_autoscaling
  • TestAccContainerNodePool_regionalAutoscaling
  • TestAccContainerNodePool_totalSize
  • TestAccContainerNodePool_withSoleTenantConfig

View the replaying VCR build log


Step 2: Recording Mode

Recording Mode Replaying Rerun Test Name
✅ Log TestAccContainerCluster_withNodePoolAutoscaling
✅ Log TestAccContainerCluster_withNodePoolCIA
✅ Log TestAccContainerNodePool_regionalAutoscaling
✅ Log TestAccContainerNodePool_totalSize
❌ Error · Log - TestAccContainerCluster_autopilot_privileged_admission
❌ Error · Log - TestAccContainerCluster_dataplaneOptimizationMode
❌ Error · Log - TestAccContainerNodePool_autoscaling
❌ Error · Log - TestAccContainerNodePool_withSoleTenantConfig

Caution

Issues requiring attention before PR completion

🔴 Initial Recording Failed: Some tests failed during the recording step. See the table above for details.

Please address these issues to complete your PR. If you believe these detections are incorrect or unrelated to your change, please raise the concern with your reviewer.

View the recording VCR build log or the debug logs folder for detailed results.

@rileykarson VCR tests complete for dabe5a8!

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants